Skip to content

chore: add kubecon na event to announcement banner - #6707

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
thisisobate:chore/kubecon-na-homepage-banner
Sep 28, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
thisisobate:chore/kubecon-na-homepage-banner

Conversation

@thisisobate

Copy link
Copy Markdown
Contributor

Proposed Changes

  • Add a KubeCon + CloudNativeCon North America 2026 to announcement banner.
  • Display the banner only on the homepage.
  • Add centered styling with official knative brand background color and white text.

Fixes #6706

Signed-off-by: thisisobate <obasiuche62@gmail.com>
@netlify

netlify Bot commented Sep 24, 2026 •

Copy link
Copy Markdown

✅ Deploy Preview for knative ready!

Built without sensitive environment variables

Name Link
🔨 Latest commit 70d87ba
🔍 Latest deploy log https://app.netlify.com/projects/knative/deploys/6ab52624304fa300074b0130
😎 Deploy Preview https://deploy-preview-6707--knative.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify project configuration.

@knative-prow knative-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 24, 2026
@knative-prow
knative-prow Bot requested review from Leo6Leo and mmejia02 September 24, 2026 13:31
@Leo6Leo

Leo6Leo commented Sep 28, 2026

Copy link
Copy Markdown
Member
CleanShot 2026-09-28 at 12 05 15@2x The banner has correctly been displayed at the top of the web page, and the link will redirect the user to a correct Kubecon page

/approve
/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 28, 2026
@knative-prow

knative-prow Bot commented Sep 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Leo6Leo, thisisobate

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 28, 2026
@knative-prow
knative-prow Bot merged commit 6b2afb1 into knative:main Sep 28, 2026
19 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Add KubeCon NA 2026 to Announcement banner

2 participants