Repository navigation
Conversation
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: tosky The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Draft PR not reviewedDraft PRs are not automatically reviewed by default.
To automatically review draft PRs, update your CodeRabbit configuration: reviews:
auto_review:
drafts: true
Comment |
|
Merge Failed. This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset. |
5c7c5c0 to
988019b
Compare
|
Build failed (check pipeline). Post ✔️ manila-s2i-content-provider SUCCESS in 20h 01m 19s |
988019b to
08f7aaa
Compare
|
Build failed (check pipeline). Post ✔️ manila-s2i-content-provider SUCCESS in 14h 48m 11s |
|
recheck |
|
Build succeeded (check pipeline). ✔️ manila-s2i-content-provider SUCCESS in 8h 19m 16s |
Depends-On: #688