Repository navigation
Conversation
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: raukadah The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Draft PR not reviewedDraft PRs are not automatically reviewed by default.
To automatically review draft PRs, update your CodeRabbit configuration: reviews:
auto_review:
drafts: trueComment |
Depends-On: openstack-k8s-operators/s2i-openstack-containers#196 Signed-off-by: Chandan <chkumar@redhat.com>
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/s2i-openstack-containers#196 is needed. |
|
recheck |
|
Build failed (check pipeline). Post ✔️ cinder-s2i-content-provider SUCCESS in 2h 46m 53s |
|
recheck |
|
Build succeeded (check pipeline). ✔️ cinder-s2i-content-provider SUCCESS in 2h 16m 47s |
Depends-On: openstack-k8s-operators/s2i-openstack-containers#196
Depends-On: openstack-k8s-operators/ci-framework#4218