Repository navigation
[WIP] Add neutron component jobs - #698
karelyatin wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: karelyatin The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
Important Review skippedIgnore keyword(s) in the title. ⛔ Ignored keywords (1)
Please check the settings in the CodeRabbit UI or the ⚙️ Run configuration
You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
b44314c to
9e0b943
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 46m 37s |
|
recheck no neutron logs |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 1h 29m 48s |
f68ff61 to
5a5b8c4
Compare
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider FAILURE in 5m 06s |
5a5b8c4 to
e2b00cd
Compare
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider FAILURE in 5m 59s |
|
recheck epel mirror |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 21m 32s |
|
recheck node failures |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 33m 48s |
e2b00cd to
bd76258
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 33m 51s |
bd76258 to
d29285d
Compare
|
Merge Failed. This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset. |
d29285d to
8c17eab
Compare
8c17eab to
6449187
Compare
|
Build failed (check pipeline). Post ❌ openstack-k8s-operators-content-provider FAILURE in 7m 22s |
|
recheck proxy error |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 5h 46m 47s |
|
recheck depends-on update required |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 8h 17m 53s |
bab6d4b to
552d72a
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 36m 55s |
c28cc6c to
81e8527
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 11h 01m 55s |
81e8527 to
4047a95
Compare
|
Merge Failed. This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset. |
4047a95 to
660dcf8
Compare
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/ci-framework#4218 is needed. |
|
depends-on update |
|
recheck depends-on update |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 37m 29s |
|
recheck depends-on updated with centos10 fix |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 22m 51s |
|
recheck depends-on selinux permissive for live migration issue |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 27m 02s |
660dcf8 to
4647534
Compare
|
@karelyatin: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
4647534 to
08b2187
Compare
|
Zuul encountered a syntax error while parsing its Job neutron-s2i-test-dvr-tls-ovn-agent not defined The problem appears in the "neutron-operator-s2i-testing" project-template stanza: project-template: in "openstack-k8s-operators/neutron-operator/zuul.d/project.yaml@main", line 3 |
Related-Issue: #OSPRH-28744 Signed-off-by: Yatin Karel <ykarel@redhat.com>
08b2187 to
2282bfb
Compare
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 5h 58m 28s |
|
recheck fixes merged and ansibleee refreshed |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 4h 45m 46s |
Related-Issue: #OSPRH-28744
Depends-On: openstack-k8s-operators/openstack-operator#2129
Depends-On: openstack-k8s-operators/s2i-openstack-containers#196
Depends-On: openstack-k8s-operators/ci-framework#4217